Skip to content

release/2.18.0 -> develop - #14897

Closed
chainchad wants to merge 9 commits into
developfrom
release/2.18.0
Closed

release/2.18.0 -> develop#14897
chainchad wants to merge 9 commits into
developfrom
release/2.18.0

Conversation

@chainchad

@chainchad chainchad commented Oct 22, 2024

Copy link
Copy Markdown
Collaborator

Signed-off-by: chainchad 96362174+chainchad@users.noreply.github.com

Needs: #15076 to fix merge conflicts.

Requires

Resolves

Signed-off-by: chainchad <96362174+chainchad@users.noreply.github.com>
@github-actions

github-actions Bot commented Oct 22, 2024

Copy link
Copy Markdown
Contributor

I see you added a changeset file but it does not contain a tag. Please edit the text include at least one of the following tags:

  • #added For any new functionality added.
  • #breaking_change For any functionality that requires manual action for the node to boot.
  • #bugfix For bug fixes.
  • #changed For any change to the existing functionality.
  • #db_update For any feature that introduces updates to database schema.
  • #deprecation_notice For any upcoming deprecation functionality.
  • #internal For changesets that need to be excluded from the final changelog.
  • #nops For any feature that is NOP facing and needs to be in the official Release Notes for the release.
  • #removed For any functionality/config that is removed.
  • #updated For any functionality that is updated.
  • #wip For any change that is not ready yet and external communication about it should be held off till it is feature complete.

🎖️ No JIRA issue number found in: PR title, commit message, or branch name. Please include the issue ID in one of these.

jmank88 and others added 4 commits October 22, 2024 14:55
…r from TypeAndVersion (#14861)

* core/services/ocr2/plugins/ccip/internal/ccipdata/factory: check error from TypeAndVersion

* .github: CCIP CODEOWNERS

(cherry picked from commit cc9bafa)
…18.0

core/services/ocr2/plugins/ccip/internal/ccipdata/factory: check error from TypeAndVersion [2.18.0]
…ent-confirmer-false-warning-2.18-cerry-pick

Hotfix/BCFR 1065 prevent confirmer false warning 2.18
@cl-sonarqube-production

Copy link
Copy Markdown

Quality Gate passed Quality Gate passed

Issues
0 New issues
0 Fixed issues
0 Accepted issues

Measures
0 Security Hotspots
No data about Coverage
No data about Duplication

See analysis details on SonarQube

ogtownsend and others added 4 commits October 30, 2024 08:55
* Update DA oracle config struct members to pointers

* Update tests and remake config docs

* suggestions and fix docs_test

* fix config test

(cherry picked from commit b585654)
…bugfix-da-struct

Update DA oracle config struct members to pointers (#15008)
Signed-off-by: chainchad <96362174+chainchad@users.noreply.github.com>
@chainchad chainchad changed the title Bump version and update CHANGELOG fore core v2.18.0 release/2.18.0 -> develop Nov 1, 2024
@chainchad

Copy link
Copy Markdown
Collaborator Author

Obviated by #15076

@chainchad chainchad closed this Nov 4, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants